Skip to content

Fix a typo in CODEOWNERS - #157356

Merged
ezio-melotti merged 1 commit into
python:mainfrom
StanFromIreland:CODEOWNDERs
Sep 12, 2026
Merged

Fix a typo in CODEOWNERS#157356
ezio-melotti merged 1 commit into
python:mainfrom
StanFromIreland:CODEOWNDERs

Conversation

@StanFromIreland

Copy link
Copy Markdown
Member

No description provided.

@ezio-melotti
ezio-melotti merged commit 227b5f0 into python:main Sep 12, 2026
41 checks passed
@StanFromIreland
StanFromIreland deleted the CODEOWNDERs branch September 12, 2026 09:37
clin1234 pushed a commit to clin1234/cpython that referenced this pull request Sep 12, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants