fix(deps): update dependency @portabletext/toolkit to v5 - #76
Open
renovate[bot] wants to merge 2 commits into
Open
renovate[bot] wants to merge 2 commits into
renovate[bot] wants to merge 2 commits into
Conversation
🦋 Changeset detectedLatest commit: 030de9b The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
|
Review the following changes in direct dependencies. Learn more about Socket for GitHub.
|
renovate
Bot
force-pushed
the
renovate/portabletext-toolkit-5.x
branch
from
September 15, 2026 22:04
f0494f4 to
7763947
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
^2.0.18→^5.0.2Release Notes
portabletext/toolkit (@portabletext/toolkit)
v5.0.2Compare Source
Patch Changes
9075772Thanks @renovate! - fix(deps): update dependency @portabletext/types to ^4.0.2v5.0.1Compare Source
Patch Changes
72963a3Thanks @renovate! - fix(deps): update dependency @portabletext/types to ^4.0.1v5.0.0Compare Source
Major Changes
cb5ee07Thanks @renovate! - fix(deps): update dependency @portabletext/types to v4v4.0.0Major Changes
#123
2852611Thanks @stipsan! - Support for node versions older than v20.19 dropped#123
2852611Thanks @stipsan! - Upgrade @portabletext/types to v3#123
2852611Thanks @stipsan! - CJS exports removedv3.0.1Compare Source
Patch Changes
12c0bbaThanks @renovate! - fix(deps): update dependency @portabletext/types to ^2.0.15v3.0.0Compare Source
Major Changes
06c95e3Thanks @stipsan! - Drop support for Sanity Studio v2Patch Changes
1eef3d9Thanks @stipsan! - update build toolingConfiguration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR has been generated by Mend Renovate using a curated preset maintained by
. View repository job log here