fix(deps): update module github.com/rakyll/statik to v0.1.8 - #990
renovate[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
ec97397 to
599158d
Compare
3011ba7 to
a66f3f2
Compare
a66f3f2 to
dc615b7
Compare
dc615b7 to
bf77096
Compare
bf77096 to
09e1899
Compare
09e1899 to
2f8d801
Compare
d446d66 to
f7e1b7b
Compare
f7e1b7b to
19d5df3
Compare
f0c855d to
69f42f1
Compare
d4b3205 to
0d4e70e
Compare
d33556a to
0f6f19f
Compare
0f6f19f to
67f45b9
Compare
b57b1a4 to
922a9f3
Compare
922a9f3 to
efa5290
Compare
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: Organization UI Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
efa5290 to
4b14cfc
Compare
This PR contains the following updates:
v0.1.7→v0.1.8Release Notes
rakyll/statik (github.com/rakyll/statik)
v0.1.8: Version 0.1.8Compare Source
What's Changed
New Contributors
Full Changelog: rakyll/statik@v0.1.7...v0.1.8
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.