fix(deps): update module github.com/spf13/cobra to v1.10.2 - #1165
renovate[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
efdddf1 to
1fe3f02
Compare
19e9e38 to
20f4571
Compare
20f4571 to
da51f24
Compare
da51f24 to
2a8b623
Compare
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: Organization UI Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
This PR contains the following updates:
v1.8.1→v1.10.2Release Notes
spf13/cobra (github.com/spf13/cobra)
v1.10.2Compare Source
🔧 Dependencies
gopkg.in/yaml.v3togo.yaml.in/yaml/v3by @dims in #2336 - thegopkg.in/yaml.v3package has been deprecated for some time: this should significantly cleanup dependency/supply-chains for consumers ofspf13/cobra📈 CI/CD
🔥✍🏼 Docs
🍂 Refactors
🤗 New Contributors
Full Changelog: spf13/cobra@v1.10.1...v1.10.2
Thank you to our amazing contributors!!!!! 🐍 🚀
v1.10.1Compare Source
🐛 Fix
v1.0.9 of pflags brought back
ParseErrorsWhitelistand marked it as deprecatedFull Changelog: spf13/cobra@v1.10.0...v1.10.1
v1.10.0Compare Source
What's Changed
🚨 Attention!
This version of
pflagcarried a breaking change: it renamedParseErrorsWhitelisttoParseErrorsAllowlistwhich can break builds if bothpflagandcobraare dependencies in your project.pflag andcobra, upgradepflagto 1.0.8 andcobrato1.10.0`pflagv1.0.9 which keeps the deprecatedParseErrorsWhitelistMore details can be found here: #2303 (comment)
✨ Features
🐛 Fix
🪠 Testing
📝 Docs
New Contributors
Full Changelog: spf13/cobra@v1.9.1...v1.9.2
v1.9.1Compare Source
🐛 Fixes
Full Changelog: spf13/cobra@v1.9.0...v1.9.1
v1.9.0Compare Source
✨ Features
🐛 Fixes
🤖 Completions
🧪 Testing
✍🏼 Documentation
🔧 Dependency upgrades
pflagdependencies for v1.9.0 by @jpmcb in #2233Thank you to all of our amazing contributors and all the great work that's been going into the completions feature!!
👋🏼 New Contributors
Full Changelog: spf13/cobra@v1.8.1...v1.9.0
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.