[main] Add Konflux configurations - #1002
Open
serverless-qe wants to merge 1 commit into
Open
serverless-qe wants to merge 1 commit into
serverless-qe wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: serverless-qe The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
serverless-qe
force-pushed
the
sync-konflux-main
branch
2 times, most recently
from
September 1, 2026 18:51
916ed3b to
00f0572
Compare
Collaborator
|
/hold |
serverless-qe
force-pushed
the
sync-konflux-main
branch
7 times, most recently
from
September 8, 2026 08:49
126b41b to
0634f4d
Compare
serverless-qe
force-pushed
the
sync-konflux-main
branch
7 times, most recently
from
September 22, 2026 08:01
947689f to
7cf40a7
Compare
serverless-qe
force-pushed
the
sync-konflux-main
branch
4 times, most recently
from
September 29, 2026 09:41
7337f52 to
2d4be13
Compare
serverless-qe
force-pushed
the
sync-konflux-main
branch
from
September 30, 2026 14:07
2d4be13 to
f5a76a2
Compare
This branch has not been deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Add Konflux components and pipelines