Skip to content

build(deps): bump graalvm/setup-graalvm from 1.6.6 to 1.6.7 in the github-actions group - #4931

Merged
kubernetes-prow[bot] merged 1 commit into
masterfrom
dependabot/github_actions/github-actions-6779201517
Oct 5, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
masterfrom
dependabot/github_actions/github-actions-6779201517

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 5, 2026

Copy link
Copy Markdown
Contributor

Bumps the github-actions group with 1 update: graalvm/setup-graalvm.

Updates graalvm/setup-graalvm from 1.6.6 to 1.6.7

Release notes

Sourced from graalvm/setup-graalvm's releases.

v1.6.7

What's Changed

Full Changelog: graalvm/setup-graalvm@v1.6.6...v1.6.7

Commits
  • 27b360d Bump version to 1.6.7.
  • 5465633 Document 25i-latest and update README.md.
  • 975bdb0 Add support for 25i-latest.
  • 311f4df Drop misleading notes.
  • 9df65ca Only run test-action-extensive with GDS_TOKEN.
  • cd60bef Bump the all group with 13 updates
  • 8b5fb99 Fix fetchArtifactByJavaVersion
  • eebbe52 Update to latest 17 CPU.
  • cce64d5 Run npm audit fix && npm run bundle.
  • f21c44e Bump undici from 6.28.0 to 6.29.0
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

Bumps the github-actions group with 1 update: [graalvm/setup-graalvm](https://github.com/graalvm/setup-graalvm).


Updates `graalvm/setup-graalvm` from 1.6.6 to 1.6.7
- [Release notes](https://github.com/graalvm/setup-graalvm/releases)
- [Commits](graalvm/setup-graalvm@0426e2e...27b360d)

---
updated-dependencies:
- dependency-name: graalvm/setup-graalvm
  dependency-version: 1.6.7
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: github-actions
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file github_actions Pull requests that update Github_actions code labels Oct 5, 2026
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Oct 5, 2026
@brendandburns

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Oct 5, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: brendandburns, dependabot[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 5, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 7f700ca into master Oct 5, 2026
3 checks passed
@dependabot
dependabot Bot deleted the dependabot/github_actions/github-actions-6779201517 branch October 5, 2026 23:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. dependencies Pull requests that update a dependency file github_actions Pull requests that update Github_actions code lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant