Skip to content

Document V8 coverage in Vitest integration - #33387

Draft
dario-piotrowicz wants to merge 1 commit into
productionfrom
dario/vitest-plugin-v8-coverage
Draft

Document V8 coverage in Vitest integration#33387
dario-piotrowicz wants to merge 1 commit into
productionfrom
dario/vitest-plugin-v8-coverage

Conversation

@dario-piotrowicz

Copy link
Copy Markdown
Member

This PR updates the workers documentation in regards to the V8 code coverage support being introduced in the vitest plugin in cloudflare/workers-sdk#15591

@github-actions

Copy link
Copy Markdown
Contributor

This pull request requires reviews from CODEOWNERS as it changes files that match the following patterns:

Pattern Owners
/src/content/docs/workers/ @GregBrimble, @irvinebroque, @korinne, @MattieTK, @mikenomitch, @vy-ton, @WalshyDev, @cloudflare/deploy-config, @cloudflare/dev-plat-leads, @cloudflare/workers-docs, @cloudflare/workers-runtime-1, @cloudflare/wrangler, @cloudflare/product-owners

@cloudflare-workers-and-pages

Copy link
Copy Markdown

🚀 Deploying Preview to Cloudflare 🚀

Preview URL: https://dario-vitest-plugin-v8-coverage.previews.developers.cloudflare.com (commit 670dffe)

This URL reflects your latest Preview deployment

Preview Deployments by commit

Status Deployment URL Commit Updated (UTC) See this deployment's details
  • Build: Success ✅
  • Deployment: Success ✅

View logs ↗
https://e949d9ae.previews.developers.cloudflare.com 670dffe 2026-09-10T22:44:21.599Z Visit the dashboard ↗

@irvinebroque

Copy link
Copy Markdown
Contributor

worth a changelog!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

product:workers Related to Workers product size/s

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants