Skip to content

Update k8s-infra-prow images as needed - #1236

Open
cert-manager-bot wants to merge 1 commit into
cert-manager:masterfrom
cert-manager-bot:autobump
Open

Update k8s-infra-prow images as needed#1236
cert-manager-bot wants to merge 1 commit into
cert-manager:masterfrom
cert-manager-bot:autobump

Conversation

@cert-manager-bot

@cert-manager-bot cert-manager-bot commented Sep 7, 2026

Copy link
Copy Markdown
Contributor

Multiple distinct us-docker.pkg.dev/k8s-infra-prow/images/ changes:

Commits Dates Images
kubernetes-sigs/prow@1a00059...e30e351 2026‑09‑04 → 2026‑09‑10 images/branchprotector, images/checkconfig, images/cherrypicker, images/clonerefs, images/crier, images/deck, images/entrypoint, images/generic-autobumper, images/ghproxy, images/hook, images/horologium, images/initupload, images/needs-rebase, images/prow-controller-manager, images/sidecar, images/sinker, images/status-reconciler, images/tide, images/tot

No gcr.io/k8s-staging-test-infra/ changes.

No europe-west1-docker.pkg.dev/cert-manager-tests-trusted/cert-manager-infra-images/ changes.

No quay.io/jetstack/base-static changes.

Nobody is currently oncall, so falling back to Blunderbuss.

@cert-manager-prow cert-manager-prow Bot added the dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. label Sep 7, 2026
@cert-manager-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign inteon for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@cert-manager-prow cert-manager-prow Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Sep 7, 2026
…ert-manager-infra-images images and quay.io/jetstack/base-static image

Multiple distinct us-docker.pkg.dev/k8s-infra-prow/images/ changes:

Commits | Dates | Images
--- | --- | ---
kubernetes-sigs/prow@1a00059...e30e351 | 2026‑09‑04 → 2026‑09‑10 | images/branchprotector, images/checkconfig, images/cherrypicker, images/clonerefs, images/crier, images/deck, images/entrypoint, images/generic-autobumper, images/ghproxy, images/hook, images/horologium, images/initupload, images/needs-rebase, images/prow-controller-manager, images/sidecar, images/sinker, images/status-reconciler, images/tide, images/tot

No gcr.io/k8s-staging-test-infra/ changes.

No europe-west1-docker.pkg.dev/cert-manager-tests-trusted/cert-manager-infra-images/ changes.

No quay.io/jetstack/base-static changes.

Signed-off-by: ci-robot <ci-robot@k8s.io>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant