ci: consolidate Parquet variant checks and fix path filters - #11153
Merged
Merged
Conversation
blaginin
commented
Sep 21, 2026
| run: cargo test -p parquet | ||
| - name: Test --all-features | ||
| run: cargo test -p parquet --all-features | ||
| - name: Test parquet-variant |
Member
Author
There was a problem hiding this comment.
this will make the check a bit slower, but it's still not the bottleneck so probably fine?
Member
Author
There was a problem hiding this comment.
plus, when we merge #11097 we can make some checks only run in the MQ and not on push
blaginin
marked this pull request as ready for review
September 21, 2026 15:29
Jefffrey
approved these changes
Sep 22, 2026
Member
|
How does this simplify the MQ? |
Member
Author
the MQ checks themselves won't change - it's more a minor cleanup to have checks more aligned with each other 🙂 |
sdf-jkl
approved these changes
Sep 22, 2026
Contributor
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
re #11097 - simplify some CI checks (so that MQ will be easier)